[main] Upgrade to latest dependencies - #4046
Conversation
|
@knative-automation: GitHub didn't allow me to request PR reviews from the following users: knative/func-approvers. Note that only knative members and repo collaborators can review this PR, and authors cannot review their own PRs. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: knative-automation The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
bumping google.golang.org/grpc 4c226da...1550d9e: > 1550d9e Change version to 1.83.1 (# 9336) > ebba6f3 Cherry-pick # 9258 and # 9332 into v1.83.x (# 9335) > 8cfeca0 Cherry-pick # 9331 to v1.83.x (# 9333) > dec6951 Change version to 1.83.1-dev (# 9229) bumping knative.dev/client/pkg 99038d9...931c46c: > 931c46c upgrade to latest dependencies (# 2267) bumping knative.dev/eventing 973bbbc...07b1e8a: > 07b1e8a build(deps): bump google.golang.org/grpc from 1.83.0 to 1.83.1 (# 9332) Signed-off-by: Knative Automation <automation@knative.team>
ef21f55 to
f1e7d42
Compare
Cron -dprotaso
/cc knative/func-approvers
/assign knative/func-approvers
Produced by: knative-extensions/knobots/actions/update-deps